[core] ci_entrypoint: fournos: don't crash in the resolve step - #267
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Navigate logical layers of code changes, visualize relationships, and explore their blast radius. Note Currently processing new changes in this PR. This may take a few minutes, please wait... ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: 📒 Files selected for processing (3)
✨ Finishing Touches🧪 Generate unit tests (beta)
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
|
/test fournos skel hellz |
🔴 Submission of
|
|
/test fournos skel hellz |
|
✅ Execution of
Execution Engine Configuration clusterless: true
exclusive: false
executionEngine:
forge:
args:
- hellz
configOverrides: {}
project: skel
owner: kpouget
pipeline: forge-test-onlyMLFlow links Pipeline Step Details ❓ 00__preflight
|
🔴 Submission of
|
|
✅ Execution of
Execution Engine Configuration clusterless: true
exclusive: false
executionEngine:
forge:
args:
- hellz
configOverrides: {}
project: skel
owner: kpouget
pipeline: forge-test-onlyMLFlow links Pipeline Step Details ❓ 00__preflight
|
|
/test fournos skel hellz |
|
✅ Execution of
Execution Engine Configuration clusterless: true
exclusive: false
executionEngine:
forge:
args:
- hellz
configOverrides: {}
project: skel
owner: kpouget
pipeline: forge-test-onlyMLFlow links Pipeline Step Details ❓ 00__preflight
|
🔴 Submission of
|
…rst pipeline step
|
/test fournos skel hellz |
|
✅ Execution of
Execution Engine Configuration clusterless: true
exclusive: false
executionEngine:
forge:
args:
- hellz
configOverrides: {}
project: skel
owner: kpouget
pipeline: forge-test-onlyMLFlow links Pipeline Step Details ❓ 00__preflight
|
🔴 Submission of
|
|
@kpouget: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/test fournos skel |
|
✅ Execution of
Execution Engine Configuration clusterless: true
exclusive: false
executionEngine:
forge:
args: []
configOverrides: {}
project: skel
owner: kpouget
pipeline: forge-test-onlyMLFlow links Pipeline Step Details ✅ 00__preflight
|
|
looks good, merging |
crashes at this level are harder to troubleshoot (logs aren't captured in MLFlow)
Summary by CodeRabbit