Skip to content

OSDOCS-20436:adds 5.0 relnotes MicroShift - #119709

Open
tedaveryredhat wants to merge 1 commit into
openshift:enterprise-5.0from
tedaveryredhat:OSDOCS-20436
Open

OSDOCS-20436:adds 5.0 relnotes MicroShift#119709
tedaveryredhat wants to merge 1 commit into
openshift:enterprise-5.0from
tedaveryredhat:OSDOCS-20436

Conversation

@tedaveryredhat

Copy link
Copy Markdown
Contributor

Version(s):
5.0

Issue:
https://redhat.atlassian.net/browse/OSDOCS-20436

Link to docs preview:

QE review:
N/A template only

Additional information:

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 10, 2026
@openshift-ci-robot

openshift-ci-robot commented Sep 10, 2026

Copy link
Copy Markdown

@tedaveryredhat: This pull request references OSDOCS-20436 which is a valid jira issue.

Details

In response to this:

Version(s):
5.0

Issue:
https://redhat.atlassian.net/browse/OSDOCS-20436

Link to docs preview:

QE review:
N/A template only

Additional information:

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Sep 10, 2026
@ocpdocs-previewbot

Copy link
Copy Markdown

🤖 Thu Sep 10 19:02:23 - Prow CI generated the docs preview:

https://119709--ocpdocs-pr.netlify.app/


See the latest images included with {microshift-short} by using the following instructions:

* xref:../microshift_updating/microshift-list-update-contents.adoc#microshift-get-rpm-release-info_microshift-list-update-contents[Listing the contents of the {microshift-short} RPM release package]

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🤖 [error] OpenShiftAsciiDoc.NoXrefInModules: Do not include xrefs in modules, only assemblies (exception: release notes modules).

See the latest images included with {microshift-short} by using the following instructions:

* xref:../microshift_updating/microshift-list-update-contents.adoc#microshift-get-rpm-release-info_microshift-list-update-contents[Listing the contents of the {microshift-short} RPM release package]
* xref:../microshift_install_bootc/microshift-install-bootc-image.adoc#microshift-install-bootc-get-published-image_microshift-install-publish-bootc-image[Getting the published bootc image for {microshift-short}]

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🤖 [error] OpenShiftAsciiDoc.NoXrefInModules: Do not include xrefs in modules, only assemblies (exception: release notes modules).

@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown

@tedaveryredhat: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/validate-portal e331c94 link true /test validate-portal

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants