Skip to content

Updated the HCM JTBD structure - #119757

Open
EricPonvelle wants to merge 1 commit into
openshift:mainfrom
EricPonvelle:job_restructure
Open

Updated the HCM JTBD structure#119757
EricPonvelle wants to merge 1 commit into
openshift:mainfrom
EricPonvelle:job_restructure

Conversation

@EricPonvelle

Copy link
Copy Markdown
Contributor

This PR updates the JTBD structures for HCM and HCM jobs.

@openshift-ci openshift-ci Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Sep 11, 2026
@ocpdocs-previewbot

ocpdocs-previewbot commented Sep 11, 2026

Copy link
Copy Markdown

🤖 Fri Sep 11 22:05:37 - Prow CI generated the docs preview:

https://119757--ocpdocs-pr.netlify.app/

@EricPonvelle
EricPonvelle force-pushed the job_restructure branch 3 times, most recently from a62e0e9 to 592f824 Compare September 11, 2026 21:25
@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

@EricPonvelle: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/validate-asciidoc b319dbb link true /test validate-asciidoc

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

branch/enterprise-4.22 branch/enterprise-5.0 size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants