[enterprise-4.16] OCPBUGS-121938: Reworked the note on Migrating a configured br-ex bri… - #120690
Conversation
|
@dfitzmau: This pull request references Jira Issue OCPBUGS-121938, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
80d5c85 to
dd38954
Compare
|
/retest |
|
@dfitzmau: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
@dfitzmau: Jira Issue OCPBUGS-121938 is in an unrecognized state (Release Pending) and will not be moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
CP of #119740 with commit ID f3c8749
Version(s):
4.16
Issue:
OCPBUGS-121938